Skip to content

Bump azure/login from 1.5.1 to 3.0.1 - #278

Closed
dependabot[bot] wants to merge 1 commit into
mainfrom
dependabot/github_actions/azure/login-3.0.1
Closed

Bump azure/login from 1.5.1 to 3.0.1#278
dependabot[bot] wants to merge 1 commit into
mainfrom
dependabot/github_actions/azure/login-3.0.1

Conversation

@dependabot

@dependabot dependabot Bot commented on behalf of github Sep 1, 2026

Copy link
Copy Markdown
Contributor

Bumps azure/login from 1.5.1 to 3.0.1.

Release notes

Sourced from azure/login's releases.

Azure Login Action v3.0.1

What's Changed

Full Changelog: https://github.com/Azure/login/commits/v3.0.1

Azure Login Action v3.0.0

What's Changed

Full Changelog: Azure/login@v2.3.0...v3.0.0

Azure Login Action v3

What's Changed

Full Changelog: Azure/login@v2.3.0...v3

Azure Login Action v2.3.1

What's Changed

Full Changelog: Azure/login@v2.3.0...v2.3.1

Azure Login Action v2.3.0

What's Changed

Full Changelog: Azure/login@v2.2.0...v2.3.0

Azure Login Action v2.2.0

What's Changed

New Contributors

... (truncated)

Commits
  • f5d393a Merge remote-tracking branch 'origin/master' into releases/v3.0.1
  • e8cd11f Replacing hardcoded version v2 with v3 (#603)
  • 73730b8 prepare release v3.0.1
  • f4dcb80 Escape single quotes in PowerShell login script inputs (#599)
  • e82415a Add IDE files to .gitignore (#597)
  • a842007 Update CODEOWNERS (#598)
  • 9dfca58 use the latest auzre/powershell@v3 (#581)
  • 893aa84 upgrade Azure Login Action version in README (#579)
  • ce6a9ff upgrade nodejs from 20 to 24 and update dependencies (#578)
  • bbcc074 Use --client-id for user-assigned managed identity authentication in Azure ...
  • Additional commits viewable in compare view

Dependabot compatibility score

Dependabot will resolve any conflicts with this PR as long as you don't alter it yourself. You can also trigger a rebase manually by commenting @dependabot rebase.


Dependabot commands and options

You can trigger Dependabot actions by commenting on this PR:

  • @dependabot rebase will rebase this PR
  • @dependabot recreate will recreate this PR, overwriting any edits that have been made to it
  • @dependabot show <dependency name> ignore conditions will show all of the ignore conditions of the specified dependency
  • @dependabot ignore this major version will close this PR and stop Dependabot creating any more for this major version (unless you reopen the PR or upgrade to it yourself)
  • @dependabot ignore this minor version will close this PR and stop Dependabot creating any more for this minor version (unless you reopen the PR or upgrade to it yourself)
  • @dependabot ignore this dependency will close this PR and stop Dependabot creating any more for this dependency (unless you reopen the PR or upgrade to it yourself)

Bumps [azure/login](https://github.com/azure/login) from 1.5.1 to 3.0.1.
- [Release notes](https://github.com/azure/login/releases)
- [Commits](Azure/login@de95379...f5d393a)

---
updated-dependencies:
- dependency-name: azure/login
  dependency-version: 3.0.1
  dependency-type: direct:production
  update-type: version-update:semver-major
...

Signed-off-by: dependabot[bot] <support@github.com>
@dependabot dependabot Bot added the ci-quality Issues that affect CI/ CD processes label Sep 1, 2026
@dependabot @github

dependabot Bot commented on behalf of github Sep 2, 2026

Copy link
Copy Markdown
Contributor Author

Superseded by #283.

@dependabot dependabot Bot closed this Sep 2, 2026
@dependabot
dependabot Bot deleted the dependabot/github_actions/azure/login-3.0.1 branch September 2, 2026 05:32
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

ci-quality Issues that affect CI/ CD processes

Projects

None yet

Development

Successfully merging this pull request may close these issues.

0 participants