Skip to content

Fix/rapidnative pill - #459

Merged
surajahmed merged 3 commits into
mainfrom
fix/rapidnative-pill
Aug 26, 2025
Merged

surajahmed merged 3 commits into
mainfrom
fix/rapidnative-pill

Conversation

@Parthgrow

Copy link
Copy Markdown
Contributor

Added the rapidnative pill in ui docs

@vercel

vercel Bot commented Aug 26, 2025 •

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

Project Deployment Preview Comments Updated (UTC)
alpha-docs Error Error Aug 26, 2025 1:52pm
2 Skipped Deployments
Project Deployment Preview Comments Updated (UTC)
nativebase-docs Ignored Ignored Aug 26, 2025 1:52pm
staging-docs Ignored Ignored Aug 26, 2025 1:52pm

@surajahmed
surajahmed merged commit e9f1a21 into main Aug 26, 2025
3 of 5 checks passed

This branch had an error being deployed

1 failed deployment
Preview – alpha-docs — 9f7cc81d Deployed Aug 26, 2025 by vercel[bot]
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants