Skip to content

repo: add @GenerateNU/infra team as CODEOWNERS - #271

Merged
Tomahawk2191 merged 1 commit into
mainfrom
isaac/codeowners
Oct 4, 2026
Merged

Tomahawk2191 merged 1 commit into
mainfrom
isaac/codeowners

Conversation

@Tomahawk2191

@Tomahawk2191 Tomahawk2191 commented Oct 4, 2026 •

Copy link
Copy Markdown
Collaborator

What

Adds @GenerateNU/infra as CODEOWNERS for the repo.

Alongside the branch ruleset, this blocks PRs from merging without approval by a member of the Infra team. Note that admins are not subject to these restrictions, and can force merge/close PRs at will.

See the CODEOWNERS docs for more details.

@fraander

fraander commented Oct 4, 2026 •

Copy link
Copy Markdown
Contributor

👷 Deploy Preview for sprightly-manatee-243873 processing.

Name Link
🔨 Latest commit ac8130a
🔍 Latest deploy log https://app.netlify.com/projects/sprightly-manatee-243873/deploys/6ac29da5189de00009496552

@Tomahawk2191 Tomahawk2191 changed the title repo: add generateNU/infra team as CODEOWNERS repo: add @GenerateNU/infra team as CODEOWNERS Oct 4, 2026
@Tomahawk2191
Tomahawk2191 merged commit 1322517 into main Oct 4, 2026
1 check passed
@Tomahawk2191
Tomahawk2191 deleted the isaac/codeowners branch October 4, 2026 18:58
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants