Skip to content

migrated nextjs from 14 to 16 & recoil to zustand - #62

Merged
Kashyap1ankit merged 1 commit into
mainfrom
fix/hero
Nov 23, 2025
Merged

Kashyap1ankit merged 1 commit into
mainfrom
fix/hero

Conversation

@Kashyap1ankit

Copy link
Copy Markdown
Owner

No description provided.

@vercel

vercel Bot commented Nov 23, 2025

Copy link
Copy Markdown
Contributor

The latest updates on your projects. Learn more about Vercel for GitHub.

1 Skipped Deployment
Project Deployment Preview Comments Updated (UTC)
jobjunction Ignored Ignored Nov 23, 2025 8:06pm

@Kashyap1ankit
Kashyap1ankit merged commit eae129a into main Nov 23, 2025
2 of 3 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant