Skip to content

feat: SDK update devnet to latest and bump rollups-node to next version - #506

Closed
brunomenezes wants to merge 4 commits into
prerelease/v2-alphafrom
feat/sdk-update-foundry-devnet-node
Closed

brunomenezes wants to merge 4 commits into
prerelease/v2-alphafrom
feat/sdk-update-foundry-devnet-node

Conversation

@brunomenezes

@brunomenezes brunomenezes commented Jul 17, 2026 •

Copy link
Copy Markdown
Member

Summary

SDK changes to support the latest devnet package and rollups-node alpha.13.

Details:

Check Steps

Replace the <tag-name> with the appropriate image tag name.

Node

docker run -ti --rm cartesi/sdk:<tag-name> cartesi-rollups-cli --version

Devnet

docker run -ti --rm cartesi/sdk:<tag-name> cat /usr/share/cartesi/DEVNET_VERSION

Foundry

docker run -ti --rm cartesi/sdk:<tag-name> anvil --version

@changeset-bot

changeset-bot Bot commented Jul 17, 2026 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 0402cab

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@cartesi/sdk Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@brunomenezes brunomenezes moved this from Todo to In progress in Rollups SDK Jul 17, 2026
@brunomenezes brunomenezes linked an issue Jul 17, 2026 that may be closed by this pull request
10 of 16 tasks
@brunomenezes
brunomenezes force-pushed the feat/sdk-update-foundry-devnet-node branch from 4418102 to bb0b33c Compare August 10, 2026 08:35
@brunomenezes

Copy link
Copy Markdown
Member Author

I am closing that one in favour of #515, which has this one as its base.

@github-project-automation github-project-automation Bot moved this from In progress to Done in Rollups SDK Aug 17, 2026
@brunomenezes
brunomenezes deleted the feat/sdk-update-foundry-devnet-node branch September 23, 2026 11:48
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

Projects

Status: Done

Development

Successfully merging this pull request may close these issues.

2 participants