feat(tab-bar-island): glass-like view - #2345
Open
hextion wants to merge 5 commits into
Open
Conversation
🦋 Changeset detectedLatest commit: fab3241 The changes in this PR will be included in the next version bump. This PR includes changesets to release 1 package
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
Contributor
Demo build (default) |
Contributor
Demo build (alfasans) |
hextion
commented
Sep 4, 2026
| .label { | ||
| @mixin action_secondary_small; | ||
|
|
||
| position: relative; |
Collaborator
Author
There was a problem hiding this comment.
to get correct z-index
hextion
force-pushed
the
feat/tab-bar-island-glass
branch
from
September 9, 2026 09:16
04eac73 to
5b467fa
Compare
hextion
marked this pull request as ready for review
September 9, 2026 10:07
hextion
requested review from
SiebenSieben,
dHIM24 and
fulcanellee
as code owners
September 9, 2026 10:07
fulcanellee
approved these changes
Sep 9, 2026
dHIM24
approved these changes
Sep 9, 2026
| height: 100%; | ||
| width: 100%; | ||
| background: var(--color-light-neutral-translucent-200); | ||
| background: var(--tab-bar-island-tab-tracker-backround-color); |
Contributor
There was a problem hiding this comment.
backround → background
Contributor
|
Заметил, что у трейлинг таба есть underlay, и он по z-index выше самой кнопки, проверь пжлст |
Contributor
hextion
force-pushed
the
feat/tab-bar-island-glass
branch
from
September 13, 2026 17:28
5b467fa to
e828332
Compare
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.

No description provided.