feat: update binutils to 2.46-2 - #21
deepin-community-bot[bot] wants to merge 3 commits into
Conversation
|
Skipping CI for Draft Pull Request. |
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
bab179c to
8560475
Compare
|
TAG Bot TAG: 2.46-2deepin2 |
|
TAG Bot New tag: 2.41-6deepin15 |
This pull request is requested by @YukariChiba.
Basic Information
Old Version: 2.41-6deepin10
New Version: 2.46-2
Old version may contain dde / deepin patches. Please review more precisely.
Potential transition
Patch series