⬆ Bump hadolint/hadolint-action from 3.4.0 to 3.5.0 - #376
dependabot[bot] wants to merge 1 commit into
Conversation
Bumps [hadolint/hadolint-action](https://github.com/hadolint/hadolint-action) from 3.4.0 to 3.5.0. - [Release notes](https://github.com/hadolint/hadolint-action/releases) - [Commits](hadolint/hadolint-action@v3.4.0...v3.5.0) --- updated-dependencies: - dependency-name: hadolint/hadolint-action dependency-version: 3.5.0 dependency-type: direct:production update-type: version-update:semver-minor ... Signed-off-by: dependabot[bot] <support@github.com>
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: dependabot[bot] The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
✅ Snyk checks have passed. No issues have been found so far.
💻 Catch issues earlier using the plugins for VS Code, JetBrains IDEs, Visual Studio, and Eclipse. |
|
Hi @dependabot[bot]. Thanks for your PR. I'm waiting for a migtools member to verify that this patch is reasonable to test. If it is, they should reply with Regular contributors should join the org to skip this step. Once the patch is verified, the new status will be reflected by the I understand the commands that are listed here. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
Bumps hadolint/hadolint-action from 3.4.0 to 3.5.0.
Release notes
Sourced from hadolint/hadolint-action's releases.
Commits
06be81bMerge pull request #107 from gizero/bump-hadolint-base-image-to-2.15.1941db07feat: Bump Hadolint to v2.15.1d0e9595Merge pull request #106 from m-ildefons/gh-78df662abProblem Matcher: Capture code and severity32c6895Merge pull request #85 from rjbell4/patch-1146301cMerge pull request #105 from m-ildefons/gh-100df8eaafCI: run integration tests with different runners75bb911fixup! Update hadolint.sh631cc83Update repository reference in problem-matcher.jsonDependabot will resolve any conflicts with this PR as long as you don't alter it yourself. You can also trigger a rebase manually by commenting
@dependabot rebase.Dependabot commands and options
You can trigger Dependabot actions by commenting on this PR:
@dependabot rebasewill rebase this PR@dependabot recreatewill recreate this PR, overwriting any edits that have been made to it@dependabot show <dependency name> ignore conditionswill show all of the ignore conditions of the specified dependency@dependabot ignore this major versionwill close this PR and stop Dependabot creating any more for this major version (unless you reopen the PR or upgrade to it yourself)@dependabot ignore this minor versionwill close this PR and stop Dependabot creating any more for this minor version (unless you reopen the PR or upgrade to it yourself)@dependabot ignore this dependencywill close this PR and stop Dependabot creating any more for this dependency (unless you reopen the PR or upgrade to it yourself)