Skip to content

Add benchmarks - #11

Merged
mspiess merged 4 commits into
mainfrom
benchmark
Sep 11, 2026
Merged

mspiess merged 4 commits into
mainfrom
benchmark

Conversation

@mspiess

@mspiess mspiess commented Aug 3, 2026

Copy link
Copy Markdown
Owner

No description provided.

@changeset-bot

changeset-bot Bot commented Aug 3, 2026

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 19a541c

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@marco.spiess/enumset Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@mspiess
mspiess force-pushed the benchmark branch 4 times, most recently from e855824 to c789b81 Compare August 9, 2026 16:28
@mspiess
mspiess force-pushed the benchmark branch 2 times, most recently from 3e1b413 to 3d14966 Compare August 17, 2026 19:34
@mspiess
mspiess merged commit 9b109dc into main Sep 11, 2026
4 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant