Skip to content

chore(deps): update registry.access.redhat.com/ubi9 docker tag to v9.8-1789646010 - #893

Open
red-hat-konflux[bot] wants to merge 1 commit into
masterfrom
konflux/mintmaker/master/registry.access.redhat.com-ubi9-9.x
Open

red-hat-konflux[bot] wants to merge 1 commit into
masterfrom
konflux/mintmaker/master/registry.access.redhat.com-ubi9-9.x

Conversation

@red-hat-konflux

@red-hat-konflux red-hat-konflux Bot commented Sep 16, 2026

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

Package Type Update Change
registry.access.redhat.com/ubi9 final patch 9.8-17893486439.8-1789646010
registry.access.redhat.com/ubi9 stage patch 9.8-17893486439.8-1789646010

Warning

Some dependencies could not be looked up. Check the Dependency Dashboard for more information.


Configuration

📅 Schedule: (in timezone UTC)

  • Branch creation
    • At any time (no schedule defined)
  • Automerge
    • At any time (no schedule defined)

🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.

Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.

🔕 Ignore: Close this PR and you won't be reminded about these updates again.


  • If you want to rebase/retry this PR, check this box

To execute skipped test pipelines write comment /ok-to-test.


Documentation

Find out how to configure dependency updates in MintMaker documentation or see all available configuration options in Renovate documentation.

@coderabbitai

coderabbitai Bot commented Sep 16, 2026

Copy link
Copy Markdown

Comment @coderabbitai help to get the list of available commands.

@joshbranham

Copy link
Copy Markdown
Contributor

/lgtm
/approve

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Sep 16, 2026
@openshift-ci

openshift-ci Bot commented Sep 16, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: joshbranham, red-hat-konflux[bot]

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Sep 16, 2026
@red-hat-konflux
red-hat-konflux Bot force-pushed the konflux/mintmaker/master/registry.access.redhat.com-ubi9-9.x branch from ec1201f to 1e71faa Compare September 17, 2026 00:14
@red-hat-konflux red-hat-konflux Bot changed the title chore(deps): update registry.access.redhat.com/ubi9 docker tag to v9.8-1789461287 chore(deps): update registry.access.redhat.com/ubi9 docker tag to v9.8-1789552280 Sep 17, 2026
@openshift-ci openshift-ci Bot removed the lgtm Indicates that a PR is ready to be merged. label Sep 17, 2026
@openshift-ci

openshift-ci Bot commented Sep 17, 2026

Copy link
Copy Markdown
Contributor

New changes are detected. LGTM label has been removed.

…8-1789646010

Signed-off-by: red-hat-konflux <126015336+red-hat-konflux[bot]@users.noreply.github.com>
@red-hat-konflux
red-hat-konflux Bot force-pushed the konflux/mintmaker/master/registry.access.redhat.com-ubi9-9.x branch from 1e71faa to 7181cfe Compare September 18, 2026 00:14
@red-hat-konflux red-hat-konflux Bot changed the title chore(deps): update registry.access.redhat.com/ubi9 docker tag to v9.8-1789552280 chore(deps): update registry.access.redhat.com/ubi9 docker tag to v9.8-1789646010 Sep 18, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant