Skip to content

1.4.0 - #181

Merged
pablopunk merged 1 commit into
mainfrom
release/1.4.0
Aug 19, 2026
Merged

1.4.0#181
pablopunk merged 1 commit into
mainfrom
release/1.4.0

Conversation

@pablopunk

Copy link
Copy Markdown
Owner

Release 1.4.0

@vercel

vercel Bot commented Aug 19, 2026 •

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

1 Skipped Deployment
Project Deployment Actions Updated (UTC)
swiftshift-app Ignored Ignored Preview Aug 19, 2026 11:09am

@codesandbox

codesandbox Bot commented Aug 19, 2026

Copy link
Copy Markdown

Review or Edit in CodeSandbox

Open the branch in Web Editor • VS Code • Insiders

Open Preview

@pablopunk
pablopunk merged commit 425d403 into main Aug 19, 2026
1 of 2 checks passed
@pablopunk
pablopunk deleted the release/1.4.0 branch August 19, 2026 11:09
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant