Skip to content

1.4.1 - #184

Merged
pablopunk merged 1 commit into
mainfrom
release/1.4.1
Aug 24, 2026
Merged

1.4.1#184
pablopunk merged 1 commit into
mainfrom
release/1.4.1

Conversation

@pablopunk

Copy link
Copy Markdown
Owner

Release 1.4.1

@codesandbox

codesandbox Bot commented Aug 24, 2026

Copy link
Copy Markdown

Review or Edit in CodeSandbox

Open the branch in Web EditorVS CodeInsiders

Open Preview

@vercel

vercel Bot commented Aug 24, 2026

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

1 Skipped Deployment
Project Deployment Actions Updated (UTC)
swiftshift-app Ignored Ignored Preview Aug 24, 2026 12:04pm

@pablopunk
pablopunk merged commit 47b942f into main Aug 24, 2026
1 check was pending
@pablopunk
pablopunk deleted the release/1.4.1 branch August 24, 2026 12:04
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant