Skip to content

Перевод раздела "Translations" на русский язык.md - #1077

Open
Midhat10 wants to merge 1 commit into
reactjs:mainfrom
Midhat10:Midhat10-patch-comminity-translation
Open

Midhat10 wants to merge 1 commit into
reactjs:mainfrom
Midhat10:Midhat10-patch-comminity-translation

Conversation

@Midhat10

Copy link
Copy Markdown

**Переведены на русский язык одна страница документации:
src/content/community/translations.md — Сведения о переводах сообщества.

@vercel

vercel Bot commented Mar 31, 2026 •

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

1 Skipped Deployment
Project Deployment Actions Updated (UTC)
ru-legacy-reactjs-org Ignored Ignored Preview Mar 31, 2026 11:44pm

Request Review

This branch was successfully deployed

1 active deployment
Preview – ru-react-dev — 95e4ab9f Deployed Mar 31, 2026 by vercel[bot]
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant